Skip to content

Bump SonarSource/sonarqube-scan-action from 8.2.2 to 8.3.0 - #976

Open
dependabot[bot] wants to merge 1 commit into
masterfrom
dependabot/github_actions/SonarSource/sonarqube-scan-action-8.3.0
Open

dependabot[bot] wants to merge 1 commit into
masterfrom
dependabot/github_actions/SonarSource/sonarqube-scan-action-8.3.0

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Oct 6, 2026

Copy link
Copy Markdown
Contributor

Bumps SonarSource/sonarqube-scan-action from 8.2.2 to 8.3.0.

Release notes

Sourced from SonarSource/sonarqube-scan-action's releases.

v8.3.0

What's Changed

New Contributors

Full Changelog: SonarSource/sonarqube-scan-action@v8.2.2...v8.3.0

Commits

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore this major version will close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this minor version will close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this dependency will close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)

Bumps [SonarSource/sonarqube-scan-action](https://github.com/sonarsource/sonarqube-scan-action) from 8.2.2 to 8.3.0.
- [Release notes](https://github.com/sonarsource/sonarqube-scan-action/releases)
- [Commits](SonarSource/sonarqube-scan-action@ba9859e...d209202)

---
updated-dependencies:
- dependency-name: SonarSource/sonarqube-scan-action
  dependency-version: 8.3.0
  dependency-type: direct:production
  update-type: version-update:semver-minor
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot added dependencies Pull requests that update a dependency file github_actions Pull requests that update GitHub Actions code labels Oct 6, 2026
@github-actions

github-actions Bot commented Oct 6, 2026

Copy link
Copy Markdown

Apache Yetus test-patch report

+1 overall
Vote Subsystem Runtime Log Comment
Prechecks
+1 dupname 0m 0s No case conflicting files found.
+1 @author 0m 0s The patch does not contain any @author tags.
master Compile Tests
Patch Compile Tests
+1 codespell 0m 2s No new issues.
+1 detsecrets 0m 33s No new issues.
+1 blanks 0m 0s The patch has no blanks issues.
+1 yamllint 0m 0s No new issues.
Other Tests
+1 asflicense 0m 1s The patch does not generate ASF License warnings.
0m 39s
Subsystem Report/Notes
GITHUB PR #976
Optional Tests dupname asflicense codespell detsecrets yamllint
uname Linux a44a9979568c 6.17.0-1022-azure #22-Ubuntu SMP Mon Jul 27 17:24:03 UTC 2026 x86_64 x86_64 x86_64 GNU/Linux
Build tool nobuild
Personality /github/workspace/.yetus/personality.sh
git revision master / 4cd657c
modules C: . U: .
versions git=2.34.1 codespell=2.4.1 detsecrets=1.5.0 yamllint=1.35.1
Powered by Apache Yetus 0.15.1 https://yetus.apache.org

This message was automatically generated.

@github-actions

github-actions Bot commented Oct 6, 2026

Copy link
Copy Markdown
TestsPassed ✅Skipped ⚠️FailedTime ⏱
JUnit Test Report442 ran436 passed6 skipped0 failed7m 34s 617ms
TestResultTime ⏱
No test annotations available

@sonarqubecloud

sonarqubecloud Bot commented Oct 6, 2026

Copy link
Copy Markdown

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file github_actions Pull requests that update GitHub Actions code

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants